Skip to content

Update Deformable Component Manipulation Readme Link ! - #1005

Open
RONAK-AI647 wants to merge 1 commit into
kubeedge:mainfrom
RONAK-AI647:RONAK-AI647/-Update-link
Open

RONAK-AI647 wants to merge 1 commit into
kubeedge:mainfrom
RONAK-AI647:RONAK-AI647/-Update-link

Conversation

@RONAK-AI647

Copy link
Copy Markdown
Contributor

What type of PR is this?

/kind documentation

What this PR does / why we need it:
Update the link to correct one !
Which issue(s) this PR fixes:

Fixes #

@kubeedge-bot kubeedge-bot added the kind/documentation Categorizes issue or PR as related to documentation. label Sep 13, 2026
@kubeedge-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: RONAK-AI647
To complete the pull request process, please assign moorezheng after the PR has been reviewed.
You can assign the PR to them by writing /assign @moorezheng in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubeedge-bot kubeedge-bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 13, 2026
Signed-off-by: RONAK-A1647 <codeitronak226277@gmail.com>
@RONAK-AI647
RONAK-AI647 force-pushed the RONAK-AI647/-Update-link branch from 4f709b7 to 44fdf27 Compare September 13, 2026 14:48
@RONAK-AI647

Copy link
Copy Markdown
Contributor Author

@MooreZheng @hsj576 , the link was broken , merge it sir please .

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/documentation Categorizes issue or PR as related to documentation. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants