Skip to content

Fix typos in README.md - #996

Open
HarshaSaiP wants to merge 1 commit into
kubeedge:mainfrom
HarshaSaiP:main
Open

HarshaSaiP wants to merge 1 commit into
kubeedge:mainfrom
HarshaSaiP:main

Conversation

@HarshaSaiP

Copy link
Copy Markdown

Corrected typos in the README file for clarity and accuracy.

/kind documentation

Corrected typos in the README file for clarity and accuracy.
@kubeedge-bot kubeedge-bot added the kind/documentation Categorizes issue or PR as related to documentation. label Sep 1, 2026
@kubeedge-bot

Copy link
Copy Markdown
Collaborator

Welcome @HarshaSaiP! It looks like this is your first PR to kubeedge/ianvs 🎉

@kubeedge-bot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: HarshaSaiP
To complete the pull request process, please assign jaypume after the PR has been reviewed.
You can assign the PR to them by writing /assign @jaypume in a comment when ready.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@kubeedge-bot kubeedge-bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 1, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/documentation Categorizes issue or PR as related to documentation. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants